Skip to content

CMP-3730: Updated banner_etc_issue to use variable - #14615

Merged
yuumasato merged 1 commit into
ComplianceAsCode:masterfrom
vickeybrown:CMP-3730-banner-fix
Aug 25, 2026
Merged

CMP-3730: Updated banner_etc_issue to use variable#14615
yuumasato merged 1 commit into
ComplianceAsCode:masterfrom
vickeybrown:CMP-3730-banner-fix

Conversation

@vickeybrown

@vickeybrown vickeybrown commented Apr 2, 2026

Copy link
Copy Markdown
Collaborator

Description:

banner_etc_issue currently uses a hardcoded remediation value, while the check allows for the configuration of different values via variable. This update changes it to follow that flow to allow for successful remediations.

This won't work without also having the changes in ComplianceAsCode/compliance-operator#1198

@openshift-ci openshift-ci Bot added the needs-ok-to-test Used by openshift-ci bot. label Apr 2, 2026
@openshift-ci

openshift-ci Bot commented Apr 2, 2026

Copy link
Copy Markdown

Hi @vickeybrown. Thanks for your PR.

I'm waiting for a ComplianceAsCode member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@yuumasato

Copy link
Copy Markdown
Member

/ok-to-test

@openshift-ci openshift-ci Bot added ok-to-test Used by openshift-ci bot. and removed needs-ok-to-test Used by openshift-ci bot. labels Apr 2, 2026
@vickeybrown

Copy link
Copy Markdown
Collaborator Author

/retest-required

@rhmdnd

rhmdnd commented Apr 6, 2026

Copy link
Copy Markdown
Collaborator

/ok-to-test

@jan-cerny jan-cerny added the OpenShift OpenShift product related. label Apr 7, 2026
@vickeybrown

Copy link
Copy Markdown
Collaborator Author

/retest

@vickeybrown
vickeybrown force-pushed the CMP-3730-banner-fix branch from 709a72d to 813621b Compare April 7, 2026 14:55
@Vincent056

Copy link
Copy Markdown
Contributor

/ok-to-test

@vickeybrown

Copy link
Copy Markdown
Collaborator Author

/test e2e-aws-openshift-node-compliance

@yuumasato yuumasato self-assigned this Apr 17, 2026
@yuumasato yuumasato added this to the 0.1.81 milestone Apr 17, 2026
@vickeybrown
vickeybrown force-pushed the CMP-3730-banner-fix branch from 813621b to b9ca2bb Compare April 23, 2026 14:32
@vickeybrown
vickeybrown force-pushed the CMP-3730-banner-fix branch 2 times, most recently from d961570 to 3d58b59 Compare May 5, 2026 17:04
@Arden97 Arden97 modified the milestones: 0.1.81, 0.1.82 May 20, 2026
@vickeybrown
vickeybrown force-pushed the CMP-3730-banner-fix branch 2 times, most recently from 04ded2e to 1604669 Compare June 3, 2026 20:38
@vickeybrown
vickeybrown force-pushed the CMP-3730-banner-fix branch from 1604669 to d9c2a14 Compare July 24, 2026 13:54
@jan-cerny jan-cerny modified the milestones: 0.1.82, 0.1.83 Aug 19, 2026

@yuumasato yuumasato left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks good, just a minor clean up on legacy remediation metadata.

@@ -5,16 +5,15 @@ kind: MachineConfig
metadata:
labels:
machineconfiguration.openshift.io/role: master

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@vickeybrown Please, remove the whole metadata block, it is not necessary and the role can be misleading, since the remediation applies on both master and worker.
CO will infer the role based on the pools it is scanning and remediating, and the MC name is inferred from the rule name

See these for example:

@openshift-ci

openshift-ci Bot commented Aug 25, 2026

Copy link
Copy Markdown

@vickeybrown: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-openshift-node-compliance d9c2a14 link true /test e2e-aws-openshift-node-compliance
ci/prow/e2e-aws-openshift-platform-compliance 3769e9d link true /test e2e-aws-openshift-platform-compliance

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@yuumasato yuumasato left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@yuumasato
yuumasato dismissed taimurhafeez’s stale review August 25, 2026 21:41

Taimur's changes have been addressed.

@yuumasato
yuumasato merged commit 5886de4 into ComplianceAsCode:master Aug 25, 2026
69 of 71 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

ok-to-test Used by openshift-ci bot. OpenShift OpenShift product related.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants