Skip to content

Update OL references - #13917

Merged
Xeicker merged 4 commits into
ComplianceAsCode:masterfrom
mrkanon:update_ol_references
Sep 23, 2025
Merged

Update OL references#13917
Xeicker merged 4 commits into
ComplianceAsCode:masterfrom
mrkanon:update_ol_references

Conversation

@mrkanon

@mrkanon mrkanon commented Sep 22, 2025

Copy link
Copy Markdown
Contributor

Description:

Update OL references to match with OL8 V2R5 and OL9 V1R2

Rationale:

To be aligned to DISA standars

Signed-off-by: Armando Acosta <armando.acosta@oracle.com>
Signed-off-by: Armando Acosta <armando.acosta@oracle.com>
Signed-off-by: Armando Acosta <armando.acosta@oracle.com>
@mrkanon
mrkanon requested a review from a team as a code owner September 22, 2025 20:36
@Xeicker

Xeicker commented Sep 22, 2025

Copy link
Copy Markdown
Contributor

Now that we are updating STIG profiles please solve the yamllint issues, see: https://complianceascode.readthedocs.io/en/latest/manual/developer/04_style_guide.html#yaml

Signed-off-by: Armando Acosta <armando.acosta@oracle.com>
@Xeicker Xeicker added this to the 0.1.79 milestone Sep 23, 2025
@Xeicker Xeicker added Oracle Linux Oracle Linux product related. STIG STIG Benchmark related. labels Sep 23, 2025
@openshift-ci

openshift-ci Bot commented Sep 23, 2025

Copy link
Copy Markdown

@mrkanon: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-openshift-node-compliance 22005bc link true /test e2e-aws-openshift-node-compliance
ci/prow/e2e-aws-openshift-platform-compliance 22005bc link true /test e2e-aws-openshift-platform-compliance

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@Xeicker Xeicker left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM,
test failures are out of the scope of this PR

@Xeicker
Xeicker merged commit 58bab45 into ComplianceAsCode:master Sep 23, 2025
135 of 138 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Oracle Linux Oracle Linux product related. STIG STIG Benchmark related.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants