Skip to content

Implement OL08-00-030645 STIG ID - #13916

Merged
Xeicker merged 2 commits into
ComplianceAsCode:masterfrom
mrkanon:implement-OL08-00-030645
Sep 23, 2025
Merged

Implement OL08-00-030645 STIG ID#13916
Xeicker merged 2 commits into
ComplianceAsCode:masterfrom
mrkanon:implement-OL08-00-030645

Conversation

@mrkanon

@mrkanon mrkanon commented Sep 22, 2025

Copy link
Copy Markdown
Contributor

Description:

Implement OL08-00-030645 STIG ID

Rationale:

New requirement was introduced in DISA STIG OL8 V2R5

Signed-off-by: Armando Acosta <armando.acosta@oracle.com>
@mrkanon
mrkanon requested a review from a team as a code owner September 22, 2025 20:28
@Xeicker Xeicker added this to the 0.1.79 milestone Sep 22, 2025
@Xeicker Xeicker added STIG STIG Benchmark related. Oracle Linux Oracle Linux product related. labels Sep 22, 2025
@Xeicker

Xeicker commented Sep 22, 2025

Copy link
Copy Markdown
Contributor

Test failure seems legit try excuting:

$ python3 utils/rule_dir_json.py
$ python3 utils/fix_rules.py sort_subkeys

Signed-off-by: Armando Acosta <armando.acosta@oracle.com>
@openshift-ci

openshift-ci Bot commented Sep 23, 2025

Copy link
Copy Markdown

@mrkanon: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-openshift-node-compliance 4190ae1 link true /test e2e-aws-openshift-node-compliance
ci/prow/e2e-aws-openshift-platform-compliance 4190ae1 link true /test e2e-aws-openshift-platform-compliance

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@Xeicker Xeicker left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM,
Failures are out of the scope of this PR

@Xeicker
Xeicker merged commit cf31f7d into ComplianceAsCode:master Sep 23, 2025
134 of 138 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Oracle Linux Oracle Linux product related. STIG STIG Benchmark related.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants