Skip to content

Fix bootc platform CPE OVAL check - #13914

Merged
Mab879 merged 1 commit into
ComplianceAsCode:masterfrom
jan-cerny:fix_bootc_platform_oval
Sep 23, 2025
Merged

Fix bootc platform CPE OVAL check#13914
Mab879 merged 1 commit into
ComplianceAsCode:masterfrom
jan-cerny:fix_bootc_platform_oval

Conversation

@jan-cerny

Copy link
Copy Markdown
Member

Curly braces aren't allowed inside curly braces. This caused that multiple rules (eg. dnf-automatic_apply_updates, mount_option_boot_noexec) were evaluated as applicable in Image Mode environments where these rules should be evaluated as notapplicable. We will fix the CPE OVAL definition and also we will synchronize the similar OVAL with the CPE OVAL.

Curly braces aren't allowed inside curly braces. This caused
that multiple rules (eg. dnf-automatic_apply_updates,
mount_option_boot_noexec) were evaluated as applicable in Image
Mode environments where these rules should be evaluated as
notapplicable. We will fix the CPE OVAL definition and also
we will synchronize the similar OVAL with the CPE OVAL.
@jan-cerny jan-cerny added this to the 0.1.79 milestone Sep 22, 2025
@jan-cerny jan-cerny added productization-issue Issue found in upstream stabilization process. CPE-AL CPE Applicability Language Image Mode Bootable containers and Image Mode RHEL labels Sep 22, 2025
@openshift-ci

openshift-ci Bot commented Sep 22, 2025

Copy link
Copy Markdown

@jan-cerny: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-openshift-node-compliance c652198 link true /test e2e-aws-openshift-node-compliance
ci/prow/e2e-aws-openshift-platform-compliance c652198 link true /test e2e-aws-openshift-platform-compliance

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@Mab879 Mab879 self-assigned this Sep 23, 2025
@Mab879
Mab879 merged commit 2e58ec5 into ComplianceAsCode:master Sep 23, 2025
135 of 138 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

CPE-AL CPE Applicability Language Image Mode Bootable containers and Image Mode RHEL productization-issue Issue found in upstream stabilization process.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants