Update rule for OL08-00-040350 - #13907
Merged
Merged
Conversation
Contributor
Author
|
/retest |
Signed-off-by: Armando Acosta <armando.acosta@oracle.com>
mrkanon
force-pushed
the
OL08-00-040350_tftp
branch
from
September 22, 2025 19:34
bde908c to
83cf60e
Compare
|
@mrkanon: The following tests failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
Xeicker
approved these changes
Sep 22, 2025
Xeicker
left a comment
Contributor
There was a problem hiding this comment.
LGTM, failed tests are unrelated to thsi PR
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Description:
Update rule for OL08-00-040350 to tftp_uses_secure_mode_systemd
Rationale:
The be aligned to OL8 STIG V2R5