Skip to content

CMP-3591: Suppress warnings about file-permissions-proxy-kubeconfig - #13894

Merged
yuumasato merged 1 commit into
ComplianceAsCode:masterfrom
rhmdnd:CMP-3591
Sep 29, 2025
Merged

CMP-3591: Suppress warnings about file-permissions-proxy-kubeconfig#13894
yuumasato merged 1 commit into
ComplianceAsCode:masterfrom
rhmdnd:CMP-3591

Conversation

@rhmdnd

@rhmdnd rhmdnd commented Sep 12, 2025

Copy link
Copy Markdown
Collaborator

This rule fetches an openshift-sdn daemon set, but if the user isn't
using that netowrking provider the operator will emit a warning about it
not being present, which is obvious and noisy.

Let's suppress it.

This rule fetches an openshift-sdn daemon set, but if the user isn't
using that netowrking provider the operator will emit a warning about it
not being present, which is obvious and noisy.

Let's suppress it.
@rhmdnd

rhmdnd commented Sep 16, 2025

Copy link
Copy Markdown
Collaborator Author

Should be ready for some reviews.

@xiaojiey

Copy link
Copy Markdown
Collaborator

/lgtm
Verification pass with content image quay.io/redhat-user-workloads/ocp-isc-tenant/compliance-operator-content-dev:on-pr-46c1107ba610240c6de7830e1e55880f2c3c663d.
Now, no warnings will prompt for a cluster with ovn netwrokType.

% oc get network cluster  -o=jsonpath={.spec.networkType}
OVNKubernetes%
% cat ssb_cis_sdn.yaml 
apiVersion: compliance.openshift.io/v1alpha1
kind: ScanSettingBinding
metadata:
  name: cis-compliance
  namespace: openshift-compliance
profiles:
  - name: upstream-ocp4-cis
    kind: Profile
    apiGroup: compliance.openshift.io/v1alpha1
settingsRef:
  name: default
  kind: ScanSetting
  apiGroup: compliance.openshift.io/v1alpha1
% oc apply -f ssb_cis_sdn.yaml 
scansettingbinding.compliance.openshift.io/cis-compliance created
% cat ssb_cis_without_fix.yaml 
apiVersion: compliance.openshift.io/v1alpha1
kind: ScanSettingBinding
metadata:
  name: cis-compliance-without-fix
  namespace: openshift-compliance
profiles:
  - name: ocp4-cis
    kind: Profile
    apiGroup: compliance.openshift.io/v1alpha1
settingsRef:
  name: default
  kind: ScanSetting
  apiGroup: compliance.openshift.io/v1alpha1
% oc apply -f ssb_cis_without_fix.yaml 
scansettingbinding.compliance.openshift.io/cis-compliance-without-fix created
% oc get scan upstream-ocp4-cis -o=jsonpath={.status.warnings}
% oc get scan ocp4-cis -o=jsonpath={.status.warnings}         
could not fetch /apis/apps/v1/namespaces/openshift-sdn/daemonsets/sdn: daemonsets.apps "sdn" not found%

@xiaojiey

Copy link
Copy Markdown
Collaborator

/lgtm

@rhmdnd rhmdnd added this to the 0.1.79 milestone Sep 19, 2025
@openshift-ci

openshift-ci Bot commented Sep 22, 2025

Copy link
Copy Markdown

@rhmdnd: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-openshift-node-compliance 46c1107 link true /test e2e-aws-openshift-node-compliance
ci/prow/e2e-aws-openshift-platform-compliance 46c1107 link true /test e2e-aws-openshift-platform-compliance

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@rhmdnd

rhmdnd commented Sep 22, 2025

Copy link
Copy Markdown
Collaborator Author

Platform and node compliance test issues are unrelated to this change and due to a recent refactor (we need to update golang versions - which we're doing in a separate patch).

@Vincent056 Vincent056 left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@yuumasato
yuumasato merged commit 162ae29 into ComplianceAsCode:master Sep 29, 2025
131 of 134 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants