CMP-3230: Update assertions for PCI-DSS testing on ARM - #13837
Merged
rhmdnd merged 1 commit intoSep 16, 2025
Conversation
rhmdnd
requested review from
Mab879,
jan-cerny,
matusmarhefka and
vojtapolasek
as code owners
August 29, 2025 19:47
yuumasato
reviewed
Sep 3, 2025
Some rules don't behave on arm64 like they do for amd64. This commit updates those so the tests pass. The is primarily due to some operators not being supported on ARM64 yet, so they don't install, making those results not-applicable.
rhmdnd
force-pushed
the
CMP-3230-update-assertions
branch
from
September 12, 2025 16:59
686f7d8 to
01641fc
Compare
Member
|
/retest |
yuumasato
approved these changes
Sep 15, 2025
Member
|
ping @Mab879 @matusmarhefka for code owners approval |
matusmarhefka
approved these changes
Sep 16, 2025
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Some rules don't behave on arm64 like they do for amd64. This commit
updates those so the tests pass. The is primarily due to some operators
not being supported on ARM64 yet, so they don't install.