Skip to content

CMP-3182: Remove unnecessary 4.2.yml assertion files - #13141

Merged
yuumasato merged 1 commit into
ComplianceAsCode:masterfrom
rhmdnd:CMP-2795
Oct 7, 2025
Merged

CMP-3182: Remove unnecessary 4.2.yml assertion files#13141
yuumasato merged 1 commit into
ComplianceAsCode:masterfrom
rhmdnd:CMP-2795

Conversation

@rhmdnd

@rhmdnd rhmdnd commented Mar 4, 2025

Copy link
Copy Markdown
Collaborator

#12216 included an extra
assertion file for 4.2.yml which we don't need. Likely a typo for
4.12.yml, which was included in the same commit.

@rhmdnd rhmdnd changed the title CMP-2795: Remove unnecessary 4.2.yml assertion files CMP-3182: Remove unnecessary 4.2.yml assertion files Mar 4, 2025
@rhmdnd
rhmdnd requested a review from yuumasato March 4, 2025 18:17
@qlty-cloud-legacy

Copy link
Copy Markdown

Code Climate has analyzed commit 822a672 and detected 0 issues on this pull request.

The test coverage on the diff in this pull request is 100.0% (50% is the threshold).

This pull request will bring the total coverage in the repository to 62.1% (0.0% change).

View more on Code Climate.

@jan-cerny jan-cerny added the OpenShift OpenShift product related. label Mar 12, 2025
@Anna-Koudelkova

Copy link
Copy Markdown
Collaborator

/lgtm

@yuumasato

Copy link
Copy Markdown
Member

@rhmdnd this needs a rebase, ;/

@rhmdnd

rhmdnd commented Sep 22, 2025

Copy link
Copy Markdown
Collaborator Author

Node and platform compliance integration tests are failing on incompatible golang versions due to a recent refactor we did in ocp4e2e. Working to address those separately.

@Vincent056 Vincent056 left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

ComplianceAsCode#12216 included an extra
assertion file for 4.2.yml which we don't need. Likely a typo for
4.12.yml, which was included in the same commit.
@openshift-ci

openshift-ci Bot commented Oct 6, 2025

Copy link
Copy Markdown

@rhmdnd: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-openshift-node-compliance a76626f link true /test e2e-aws-openshift-node-compliance

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@Mab879 Mab879 added this to the 0.1.79 milestone Oct 6, 2025
@yuumasato yuumasato self-assigned this Oct 7, 2025
@yuumasato
yuumasato merged commit 6ab30b6 into ComplianceAsCode:master Oct 7, 2025
133 of 136 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

OpenShift OpenShift product related.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants